Skip to content

Build MinIO test image from source - #2116

Merged
openshift-merge-bot[bot] merged 1 commit into
backube:mainfrom
tesshuflower:minio-source-build-e2e
Sep 30, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
backube:mainfrom
tesshuflower:minio-source-build-e2e

Conversation

@tesshuflower

@tesshuflower tesshuflower commented Sep 29, 2026 •

Copy link
Copy Markdown
Contributor
  • minio has stopped making built images available, see notes about it being a source-only distribution now: https://github.com/tesshuflower/volsync/pull/new/minio-source-build-e2e
  • build a minio image directly for e2e and local dev called 'volsync-test-storage'.
  • for custom scorecard tests where we also need minio, pull the binaries from the 'volsync-test-storage' image into the scorecard image itself so scorecard image can be used to run minio.
  • To keep this change minimal for the moment, not changing the minio helm charts themselves, just the image we reference.

Describe what this PR does

Is there anything that requires special attention?

We will still need a pr to openshift/release for the openshift e2e CI tests run by prow. These will also need to be updated to build minio.

Related issues:

@openshift-ci

openshift-ci Bot commented Sep 29, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: tesshuflower

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Comment thread Dockerfile.minio-test-storage Fixed
- minio has stopped making built images available
- build a minio image directly for e2e and local dev called
  'volsync-test-storage'.
- for custom scorecard tests where we also need minio,
  pull the binaries from the 'volsync-test-storage' image into
  the scorecard image itself so scorecard image can be used to
  run minio.
- To keep this change minimal for the moment, not changing the
  minio helm charts themselves, just the image we reference.

Signed-off-by: Tesshu Flower <tflower@redhat.com>
@sonarqubecloud

Copy link
Copy Markdown

@JohnStrunk

Copy link
Copy Markdown
Member

/lgtm

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/override "ci/prow/e2e-openshift"

Will need updates in prow to build minio for openshift e2e tests - so overriding the check

@openshift-ci

openshift-ci Bot commented Sep 30, 2026

Copy link
Copy Markdown
Contributor

@tesshuflower: Overrode contexts on behalf of tesshuflower: ci/prow/e2e-openshift

Details

In response to this:

/override "ci/prow/e2e-openshift"

Will need updates in prow to build minio for openshift e2e tests - so overriding the check

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@tesshuflower

Copy link
Copy Markdown
Contributor Author

/override "ci/prow/e2e-openshift-fips"
/override "ci/prow/e2e-openshift-helm"

@openshift-ci

openshift-ci Bot commented Sep 30, 2026

Copy link
Copy Markdown
Contributor

@tesshuflower: Overrode contexts on behalf of tesshuflower: ci/prow/e2e-openshift-fips, ci/prow/e2e-openshift-helm

Details

In response to this:

/override "ci/prow/e2e-openshift-fips"
/override "ci/prow/e2e-openshift-helm"

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@codecov

codecov Bot commented Sep 30, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 68.7%. Comparing base (b2ea365) to head (e3eb9a4).
⚠️ Report is 2 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##            main   #2116     +/-   ##
=======================================
- Coverage   68.8%   68.7%   -0.1%     
=======================================
  Files         58      58             
  Lines       6076    6076             
=======================================
- Hits        4184    4178      -6     
- Misses      1582    1585      +3     
- Partials     310     313      +3     

see 1 file with indirect coverage changes

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@openshift-merge-bot
openshift-merge-bot Bot merged commit e3e67f1 into backube:main Sep 30, 2026
65 of 66 checks passed
@tesshuflower
tesshuflower deleted the minio-source-build-e2e branch September 30, 2026 15:28
tesshuflower added a commit to tesshuflower/volsync that referenced this pull request Sep 30, 2026
backport of: backube#2116

Signed-off-by: Tesshu Flower <tflower@redhat.com>
tesshuflower added a commit to tesshuflower/volsync that referenced this pull request Sep 30, 2026
backport of: backube#2116

Signed-off-by: Tesshu Flower <tflower@redhat.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants