chore(deps): update renovate/renovate docker tag to v44.79.2 - #742
Merged
Merged
Conversation
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 11, 2026 11:30
d5fbc80 to
0c18931
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 11, 2026 18:08
0c18931 to
b90dee9
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 11, 2026 22:02
b90dee9 to
41c9945
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 00:21
41c9945 to
211b44d
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 04:28
211b44d to
0692355
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 08:25
0692355 to
0a1d656
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 12:21
0a1d656 to
8afaa5b
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 16:22
8afaa5b to
e7c13f7
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 12, 2026 20:24
e7c13f7 to
83f7cc4
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 00:32
83f7cc4 to
bb50268
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 04:23
bb50268 to
741f200
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 08:23
741f200 to
621c8af
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 12:22
621c8af to
118ecff
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 16:24
118ecff to
5ab5179
Compare
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: erikgb The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 13, 2026 17:43
5ab5179 to
37417ca
Compare
Contributor
|
New changes are detected. LGTM label has been removed. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
44.69.5→44.79.244.83.0(+12)Release Notes
renovatebot/renovate (renovate/renovate)
v44.79.2Compare Source
Bug Fixes
Miscellaneous Chores
v44.79.1Compare Source
Bug Fixes
common-tagsto a prod dependency (#45850) (1d33143)v44.79.0Compare Source
Features
inheritConfigTrustedto trust inherited-confighostRules(238e15a)allowInternal(09d2da2)Bug Fixes
enabledper trust tier (7d3fe44)Code Refactoring
Tests
v44.78.0Compare Source
Features
eas-version(#45846) (6d67f70)Bug Fixes
Code Refactoring
v44.77.0Compare Source
Features
exitCodeForErrorsoption (#45752) (9cecb86)v44.76.0Compare Source
Features
Miscellaneous Chores
Tests
Continuous Integration
v44.75.2Compare Source
Build System
Continuous Integration
v44.75.1Compare Source
Build System
v44.75.0Compare Source
Features
v44.74.1Compare Source
Bug Fixes
Miscellaneous Chores
v44.74.0Compare Source
Features
Miscellaneous Chores
v44.73.0Compare Source
Features
Miscellaneous Chores
v44.72.0Compare Source
Features
Bug Fixes
platform=local(#45769) (f2896de)Code Refactoring
Build System
v44.71.3Compare Source
Miscellaneous Chores
Code Refactoring
Build System
v44.71.2Compare Source
Miscellaneous Chores
Build System
v44.71.1Compare Source
Bug Fixes
v44.71.0Compare Source
Features
Bug Fixes
Miscellaneous Chores
Build System
v44.70.2Compare Source
Build System
v44.70.1Compare Source
Build System
v44.70.0Compare Source
Features
Miscellaneous Chores
v44.69.14Compare Source
Build System
v44.69.13Compare Source
Build System
v44.69.12Compare Source
Bug Fixes
Miscellaneous Chores
v44.69.11Compare Source
Bug Fixes
Documentation
Miscellaneous Chores
v44.69.10Compare Source
Bug Fixes
v44.69.9Compare Source
Bug Fixes
Documentation
v44.69.8Compare Source
Bug Fixes
.npmrcsanitization (#44652) (2ee9a26)v44.69.7Compare Source
Bug Fixes
v44.69.6Compare Source
Build System
Configuration
📅 Schedule: (in timezone Europe/London)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.